Skip to content

translate/reference/react dom flushSync - #670

Closed
deblasis wants to merge 2 commits into
translate/reference/react-dom-createPortalfrom
translate/reference/react-dom-flushSync
Closed

deblasis wants to merge 2 commits into
translate/reference/react-dom-createPortalfrom
translate/reference/react-dom-flushSync

Conversation

@deblasis

Copy link
Copy Markdown
Collaborator

No description provided.

@github-actions

github-actions Bot commented Sep 14, 2026 •

Copy link
Copy Markdown

Size changes

Details

📦 Next.js Bundle Analysis for react-dev

This analysis was generated by the Next.js Bundle Analysis action. 🤖

This PR introduced no changes to the JavaScript bundle! 🙌

@deblasis
deblasis force-pushed the translate/reference/react-dom-flushSync branch from bb6c06c to 15838af Compare September 14, 2026 15:13
@vercel

vercel Bot commented Sep 14, 2026 •

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for GitHub.

1 Skipped Deployment
Project Deployment Actions Updated
legacy-it-reactjs-org Ignored Ignored Preview Sep 14, 2026 8:51pm UTC

Request Review

@deblasis
deblasis marked this pull request as ready for review September 14, 2026 15:16
@deblasis
deblasis force-pushed the translate/reference/react-dom-flushSync branch from 15838af to 5f07a89 Compare September 14, 2026 15:23
deblasis added a commit that referenced this pull request Sep 14, 2026
Signed-off-by: Alessandro De Blasis <alex@deblasis.net>
deblasis added a commit that referenced this pull request Sep 14, 2026
Signed-off-by: Alessandro De Blasis <alex@deblasis.net>
deblasis added a commit that referenced this pull request Sep 14, 2026
Signed-off-by: Alessandro De Blasis <alex@deblasis.net>
@deblasis
deblasis force-pushed the translate/reference/react-dom-flushSync branch from c69bc79 to 5d088cd Compare September 14, 2026 17:29
deblasis added a commit that referenced this pull request Sep 14, 2026
Signed-off-by: Alessandro De Blasis <alex@deblasis.net>
deblasis added a commit that referenced this pull request Sep 14, 2026
Signed-off-by: Alessandro De Blasis <alex@deblasis.net>
@deblasis
deblasis force-pushed the translate/reference/react-dom-flushSync branch from 5d088cd to a02c1f6 Compare September 14, 2026 17:41
deblasis added a commit that referenced this pull request Sep 14, 2026
Signed-off-by: Alessandro De Blasis <alex@deblasis.net>
deblasis added a commit that referenced this pull request Sep 14, 2026
Signed-off-by: Alessandro De Blasis <alex@deblasis.net>
@deblasis
deblasis force-pushed the translate/reference/react-dom-flushSync branch from a02c1f6 to 1ca12b0 Compare September 14, 2026 18:32
deblasis added a commit that referenced this pull request Sep 14, 2026
Signed-off-by: Alessandro De Blasis <alex@deblasis.net>
deblasis added a commit that referenced this pull request Sep 14, 2026
Signed-off-by: Alessandro De Blasis <alex@deblasis.net>
deblasis added a commit that referenced this pull request Sep 14, 2026
Signed-off-by: Alessandro De Blasis <alex@deblasis.net>
deblasis added a commit that referenced this pull request Sep 14, 2026
Signed-off-by: Alessandro De Blasis <alex@deblasis.net>
deblasis added a commit that referenced this pull request Sep 14, 2026
Signed-off-by: Alessandro De Blasis <alex@deblasis.net>
deblasis added a commit that referenced this pull request Sep 14, 2026
Signed-off-by: Alessandro De Blasis <alex@deblasis.net>
deblasis added a commit that referenced this pull request Sep 14, 2026
Signed-off-by: Alessandro De Blasis <alex@deblasis.net>
deblasis added a commit that referenced this pull request Sep 14, 2026
Signed-off-by: Alessandro De Blasis <alex@deblasis.net>
@deblasis
deblasis force-pushed the translate/reference/react-dom-flushSync branch from 1ca12b0 to 1eb54fd Compare September 14, 2026 19:51
Signed-off-by: Alessandro De Blasis <alex@deblasis.net>
Signed-off-by: Alessandro De Blasis <alex@deblasis.net>
@deblasis
deblasis force-pushed the translate/reference/react-dom-flushSync branch from 1eb54fd to a9936fd Compare September 14, 2026 20:51
@deblasis deblasis closed this Sep 14, 2026
deblasis added a commit that referenced this pull request Sep 14, 2026
Signed-off-by: Alessandro De Blasis <alex@deblasis.net>

This branch was successfully deployed

1 active deployment
Preview – it-react-dev — a9936fd2 Deployed Sep 14, 2026 by vercel[bot]
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant