Skip to content

translate/reference/react dom createPortal - #669

Closed
deblasis wants to merge 2 commits into
translate/reference/react-dom-textareafrom
translate/reference/react-dom-createPortal
Closed

deblasis wants to merge 2 commits into
translate/reference/react-dom-textareafrom
translate/reference/react-dom-createPortal

Conversation

@deblasis

Copy link
Copy Markdown
Collaborator

No description provided.

@github-actions

github-actions Bot commented Sep 14, 2026 •

Copy link
Copy Markdown

Size changes

Details

📦 Next.js Bundle Analysis for react-dev

This analysis was generated by the Next.js Bundle Analysis action. 🤖

This PR introduced no changes to the JavaScript bundle! 🙌

@deblasis
deblasis force-pushed the translate/reference/react-dom-createPortal branch from 5ebd1e9 to 7ce43a2 Compare September 14, 2026 15:13
@vercel

vercel Bot commented Sep 14, 2026 •

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for GitHub.

1 Skipped Deployment
Project Deployment Actions Updated
legacy-it-reactjs-org Ignored Ignored Preview Sep 14, 2026 8:51pm UTC

Request Review

Signed-off-by: Alessandro De Blasis <alex@deblasis.net>
Signed-off-by: Alessandro De Blasis <alex@deblasis.net>
@deblasis
deblasis force-pushed the translate/reference/react-dom-createPortal branch from a0b50b4 to 4b031d2 Compare September 14, 2026 20:51
@deblasis deblasis closed this Sep 14, 2026
deblasis added a commit that referenced this pull request Sep 14, 2026
Signed-off-by: Alessandro De Blasis <alex@deblasis.net>

This branch was successfully deployed

1 active deployment
Preview – it-react-dev — 4b031d2a Deployed Sep 14, 2026 by vercel[bot]
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant