Skip to content

translate/reference/useSyncExternalStore - #652

Merged
deblasis merged 1 commit into
translate/reference/useStatefrom
translate/reference/useSyncExternalStore
Sep 14, 2026
Merged

deblasis merged 1 commit into
translate/reference/useStatefrom
translate/reference/useSyncExternalStore

Conversation

@deblasis

Copy link
Copy Markdown
Collaborator

No description provided.

@github-actions

github-actions Bot commented Sep 14, 2026 •

Copy link
Copy Markdown

Size changes

Details

📦 Next.js Bundle Analysis for react-dev

This analysis was generated by the Next.js Bundle Analysis action. 🤖

One Page Changed Size

The following page changed size from the code in this PR compared to its base branch:

Page Size (compressed) First Load
/[[...markdownPath]] 130.69 KB (-5 B) 245.84 KB
Details

Only the gzipped size is provided here based on an expert tip.

First Load is the size of the global bundle plus the bundle for the individual page. If a user were to show up to your website and land on a given page, the first load size represents the amount of javascript that user would need to download. If next/link is used, subsequent page loads would only need to download that page's bundle (the number in the "Size" column), since the global bundle has already been downloaded.

Any third party scripts you have added directly to your app using the <script> tag are not accounted for in this analysis

Next to the size is how much the size has increased or decreased compared with the base branch of this PR. If this percentage has increased by 10% or more, there will be a red status indicator applied, indicating that special attention should be given to this.

@deblasis
deblasis force-pushed the translate/reference/useSyncExternalStore branch from de6c1a8 to 1d753e5 Compare September 14, 2026 15:12
@vercel

vercel Bot commented Sep 14, 2026 •

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for GitHub.

1 Skipped Deployment
Project Deployment Actions Updated
legacy-it-reactjs-org Ignored Ignored Preview Sep 14, 2026 7:48pm UTC

Request Review

@deblasis
deblasis marked this pull request as ready for review September 14, 2026 15:15
deblasis added a commit that referenced this pull request Sep 14, 2026
Signed-off-by: Alessandro De Blasis <alex@deblasis.net>
deblasis added a commit that referenced this pull request Sep 14, 2026
Signed-off-by: Alessandro De Blasis <alex@deblasis.net>
@deblasis
deblasis force-pushed the translate/reference/useSyncExternalStore branch from 9943c26 to e693418 Compare September 14, 2026 17:40
deblasis added a commit that referenced this pull request Sep 14, 2026
Signed-off-by: Alessandro De Blasis <alex@deblasis.net>
@deblasis
deblasis force-pushed the translate/reference/useSyncExternalStore branch from e693418 to ce86f9c Compare September 14, 2026 18:31
Signed-off-by: Alessandro De Blasis <alex@deblasis.net>
@deblasis
deblasis force-pushed the translate/reference/useSyncExternalStore branch from ce86f9c to 1d4e634 Compare September 14, 2026 19:48
@deblasis
deblasis merged commit 95d9ecd into main Sep 14, 2026
10 checks passed

This branch was successfully deployed

1 active deployment
Preview – it-react-dev — 1d4e6347 Deployed Sep 14, 2026 by vercel[bot]
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant