Skip to content

translate/reference/useEffect - #643

Merged
deblasis merged 1 commit into
translate/reference/useDeferredValuefrom
translate/reference/useEffect
Sep 14, 2026
Merged

deblasis merged 1 commit into
translate/reference/useDeferredValuefrom
translate/reference/useEffect

Conversation

@deblasis

Copy link
Copy Markdown
Collaborator

No description provided.

@github-actions

Copy link
Copy Markdown

Size changes

Details

📦 Next.js Bundle Analysis for react-dev

This analysis was generated by the Next.js Bundle Analysis action. 🤖

This PR introduced no changes to the JavaScript bundle! 🙌

@deblasis
deblasis force-pushed the translate/reference/useEffect branch from df4d110 to ddc9ab0 Compare September 14, 2026 15:00
@vercel

vercel Bot commented Sep 14, 2026 •

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for GitHub.

1 Skipped Deployment
Project Deployment Actions Updated
legacy-it-reactjs-org Ignored Ignored Preview Sep 14, 2026 3:11pm UTC

Request Review

Signed-off-by: Alessandro De Blasis <alex@deblasis.net>
@deblasis
deblasis force-pushed the translate/reference/useEffect branch from ddc9ab0 to 262b129 Compare September 14, 2026 15:11
@deblasis
deblasis marked this pull request as ready for review September 14, 2026 15:15
@deblasis
deblasis merged commit 262b129 into main Sep 14, 2026
12 checks passed
deblasis added a commit that referenced this pull request Sep 14, 2026
Signed-off-by: Alessandro De Blasis <alex@deblasis.net>
deblasis added a commit that referenced this pull request Sep 14, 2026
Signed-off-by: Alessandro De Blasis <alex@deblasis.net>
deblasis added a commit that referenced this pull request Sep 14, 2026
Signed-off-by: Alessandro De Blasis <alex@deblasis.net>

This branch was successfully deployed

1 active deployment
Preview – it-react-dev — 262b129e Deployed Sep 14, 2026 by vercel[bot]
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant